Update module go.uber.org/zap to v1.28.0 - #1990
renovate-bot wants to merge 1 commit into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: renovate-bot The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
8e16a9f to
faa682e
Compare
|
faa682e to
0b8c717
Compare
|
Signed-off-by: Mend Renovate <bot@renovateapp.com>
0b8c717 to
bb85fec
Compare
|
Codecov Report✅ All modified and coverable lines are covered by tests. Additional details and impacted files@@ Coverage Diff @@
## main #1990 +/- ##
=======================================
- Coverage 68.7% 68.6% -0.1%
=======================================
Files 58 58
Lines 6056 6056
=======================================
- Hits 4164 4158 -6
- Misses 1581 1584 +3
- Partials 311 314 +3 🚀 New features to boost your workflow:
|
|
This pull request has been automatically marked as stale because it has not had recent activity. It will be closed in 30 days if no further activity occurs. Thank you for your contributions. |



This PR contains the following updates:
v1.27.1→v1.28.0Release Notes
uber-go/zap (go.uber.org/zap)
v1.28.0Compare Source
Enhancements:
zapcore.CheckPreWriteHookandCheckedEntry.Beforemethod for transforming entries before they are written to any Cores.Configuration
📅 Schedule: (in timezone America/New_York)
* 3-6 * * 1-5)🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR was generated by Mend Renovate. View the repository job log.