Update module k8s.io/component-helpers to v0.37.1 - #2035
renovate-bot wants to merge 1 commit into
Conversation
|
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: renovate-bot The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
512bf75 to
3c04a61
Compare
|
|
@renovate-bot: The following tests failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
This pull request has been automatically marked as stale because it has not had recent activity. It will be closed in 30 days if no further activity occurs. Thank you for your contributions. |
3c04a61 to
8dcecfc
Compare
8dcecfc to
81bdc96
Compare
81bdc96 to
3df5f83
Compare
Signed-off-by: Mend Renovate <bot@renovateapp.com>
3df5f83 to
1ab1c7a
Compare
|



This PR contains the following updates:
v0.35.2→v0.37.1Release Notes
kubernetes/component-helpers (k8s.io/component-helpers)
v0.37.1Compare Source
v0.37.0Compare Source
v0.36.5Compare Source
v0.36.4Compare Source
v0.36.3Compare Source
v0.36.2Compare Source
v0.36.1Compare Source
v0.36.0Compare Source
v0.35.9Compare Source
v0.35.8Compare Source
v0.35.7Compare Source
v0.35.6Compare Source
v0.35.5Compare Source
v0.35.4Compare Source
v0.35.3Compare Source
Configuration
📅 Schedule: (in timezone America/New_York)
* 3-6 * * 1-5)🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR was generated by Mend Renovate. View the repository job log.